Skip to content

Navigation Menu

Sign in
Sign up

Перевод на русский компонента "Footer" - #1034

Merged
titovmx merged 1 commit into
reactjs:main from
NovikovIlya:patch-1
Nov 11, 2025
Merged

Перевод на русский компонента "Footer" #1034
titovmx merged 1 commit into
reactjs:main from
NovikovIlya:patch-1

Conversation

@NovikovIlya

@NovikovIlya NovikovIlya commented Aug 1, 2025

Copy link
Copy Markdown
Contributor

Перевод на русский компонента "Footer"

vercel Bot commented Aug 1, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for Git ↗︎

1 Skipped Deployment
Name Status Preview Comments Updated (UTC)
ru-legacy-reactjs-org ⬜️ Ignored (Inspect) Aug 1, 2025 7:30am

titovmx merged commit 0491012 into reactjs:main Nov 11, 2025
7 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Reviewers

@titovmx titovmx titovmx approved these changes

Assignees

No one assigned

Labels

None yet

Projects

None yet

Milestone

No milestone

Development

Successfully merging this pull request may close these issues.

AltStyle によって変換されたページ (->オリジナル) /