Skip to content

Navigation Menu

Sign in
Sign up

chore: remove unsafe fallback in calculate_external_state_root - #481

Open
alessandromazza98 wants to merge 5 commits into
flashbots:main from
alessandromazza98:ale/bp2
Open

chore: remove unsafe fallback in calculate_external_state_root #481
alessandromazza98 wants to merge 5 commits into
flashbots:main from
alessandromazza98:ale/bp2

Conversation

@alessandromazza98

@alessandromazza98 alessandromazza98 commented Apr 7, 2026

Copy link
Copy Markdown
Contributor

Stacked on top of #480

This closes best practice 2 of Nethermind security audit

vercel Bot commented Apr 7, 2026
edited
Loading

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
rollup-boost Ignored Ignored Preview Apr 7, 2026 11:06am

Request Review

sleep_until(Instant::now() + self.health_check_interval).await;
continue;
} else if self.execution_mode.lock().is_disabled()
|| self.execution_mode.lock().is_dry_run()

@0xForerunner 0xForerunner Apr 9, 2026

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Are we sure we want to remove the dry run logic here?

@alessandromazza98 alessandromazza98 Apr 10, 2026

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nethermind audit suggestion says to treat dry_run and enabled in the same way for this specific thing

Err(e) => {
warn!(target: "rollup_boost::health", "Builder client - Failed to get unsafe block {} - updating health status", e);
self.probes.set_health(Health::PartialContent);
// In enabled and dry-run modes the builder still matters for health.

@0xForerunner 0xForerunner Apr 9, 2026

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I'm not sure I understand the logic here for dry run?

@alessandromazza98 alessandromazza98 Apr 10, 2026

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

essentially for both enabled and dry-run modes, you need to check both the builder AND the l2 node

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Reviewers

@0xForerunner 0xForerunner 0xForerunner left review comments

At least 1 approving review is required to merge this pull request.

Assignees

No one assigned

Labels

None yet

Projects

None yet

Milestone

No milestone

Development

Successfully merging this pull request may close these issues.

AltStyle によって変換されたページ (->オリジナル) /