Skip to content

Navigation Menu

Sign in
Sign up

fix spec metadata structure - #417

Open
meerkat-monkey wants to merge 2 commits into
flashbots:main from
meerkat-monkey:fix/spec-metadata-structure
Open

fix spec metadata structure #417
meerkat-monkey wants to merge 2 commits into
flashbots:main from
meerkat-monkey:fix/spec-metadata-structure

Conversation

@meerkat-monkey

@meerkat-monkey meerkat-monkey commented Sep 30, 2025

Copy link
Copy Markdown

The FlashblockMetadata described in the spec is different from the implementation. I assumed the implementation is correct, and have updated the spec to reflect the implementation.

Transaction receipts included in the flashblocks metadata are serialized from the following enum defined here.

pub enum OpReceipt {
 /// Legacy receipt
 Legacy(Receipt),
 /// EIP-2930 receipt
 Eip2930(Receipt),
 /// EIP-1559 receipt
 Eip1559(Receipt),
 /// EIP-7702 receipt
 Eip7702(Receipt),
 /// Deposit receipt
 Deposit(OpDepositReceipt),
}

This does not translate to Python succinctly so I opted to show it in Rust. I think that it makes sense to update the spec to show all of the types in Rust to make it easy to translate to the code stored in the repo, and I'm happy to make that change if it's agreed.

vercel Bot commented Sep 30, 2025
edited
Loading

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Preview Comments Updated (UTC)
rollup-boost Ignored Ignored Preview Sep 30, 2025 6:17pm

akundaz commented Sep 30, 2025

Copy link
Copy Markdown
Contributor

Thanks for the fix!

Can you change code pointers to point to a specific commit so that the links still work even if the code moves around?

--chain=optimism \
./target/release/flashblocks-rpc node \
--flashblocks.enabled \
--flashblocks.websocket-url="wss://sepolia-flashblocks.unichain.org/ws" \

@avalonche avalonche Sep 30, 2025

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

would prefer to have the local node here instead of pointing to unichain in the readme

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Reviewers

@avalonche avalonche avalonche left review comments

At least 1 approving review is required to merge this pull request.

Assignees

No one assigned

Labels

None yet

Projects

None yet

Milestone

No milestone

Development

Successfully merging this pull request may close these issues.

AltStyle によって変換されたページ (->オリジナル) /